Skip to content

Version Packages - #4020

Open
github-actions[bot] wants to merge 1 commit into
masterfrom
changeset-release/master
Open

github-actions[bot] wants to merge 1 commit into
masterfrom
changeset-release/master

Conversation

@github-actions

@github-actions github-actions Bot commented Jul 12, 2026 •

Copy link
Copy Markdown
Contributor

This PR was opened by the Changesets release GitHub action. When you're ready to do a release, you can merge this and the packages will be published to npm automatically. If you're not ready to do a release yet, that's fine, whenever you add more changesets to master, this PR will be updated.

Releases

The changelog information of each package has been omitted from this message, as the content exceeds the size limit.

@data-client/core@1.0.0

@data-client/endpoint@1.0.0

@data-client/graphql@1.0.0

@data-client/img@1.0.0

@data-client/normalizr@1.0.0

@data-client/react@1.0.0

@data-client/test@1.0.0

@data-client/rest@1.0.0

@data-client/vue@1.0.0

example-benchmark@0.4.87

example-benchmark-react@0.1.12

coinbase-lite@0.0.27

normalizr-github-example@0.1.61

normalizr-redux-example@0.1.59

normalizr-relationships@0.1.61

test-bundlesize@0.1.19

@vercel

vercel Bot commented Jul 12, 2026 •

Copy link
Copy Markdown

The latest updates on your projects. Learn more about Vercel for GitHub.

1 Skipped Deployment
Project Deployment Actions Updated
docs-site Ignored Ignored Preview Oct 6, 2026 4:22pm UTC

Request Review

@github-actions
github-actions Bot force-pushed the changeset-release/master branch from 4a424c3 to 5b15769 Compare July 12, 2026 13:04
@codecov

codecov Bot commented Jul 12, 2026 •

Copy link
Copy Markdown

Codecov Report

✅ All modified and coverable lines are covered by tests.
✅ Project coverage is 98.10%. Comparing base (ee826dc) to head (48a3a14).

Additional details and impacted files
@@           Coverage Diff           @@
##           master    #4020   +/-   ##
=======================================
  Coverage   98.10%   98.10%           
=======================================
  Files         166      166           
  Lines        3166     3166           
  Branches      626      626           
=======================================
  Hits         3106     3106           
  Misses         18       18           
  Partials       42       42           

☔ View full report in Codecov by Harness.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.
  • 📦 JS Bundle Analysis: Save yourself from yourself by tracking and limiting bundle sizes in JS merges.

@github-actions
github-actions Bot force-pushed the changeset-release/master branch 15 times, most recently from 1f46e28 to f419c83 Compare July 18, 2026 19:33
@github-actions
github-actions Bot force-pushed the changeset-release/master branch 12 times, most recently from 80424db to eb5f7a5 Compare July 25, 2026 17:39
@github-actions
github-actions Bot force-pushed the changeset-release/master branch 29 times, most recently from ffe669a to e3522d3 Compare October 4, 2026 18:24

This branch was successfully deployed

1 active (outdated) deployment
Preview — 55011e6e Deployed Oct 5, 2026 by vercel[bot]
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants