Skip to content

Administer and Disconnected map - #119840

Open
abhatt-rh wants to merge 3 commits into
openshift:mainfrom
abhatt-rh:administer-map
Open

abhatt-rh wants to merge 3 commits into
openshift:mainfrom
abhatt-rh:administer-map

Conversation

@abhatt-rh

Copy link
Copy Markdown
Contributor

Version(s):

Issue:

Link to docs preview:

QE review:

  • QE has approved this change.

Additional information:

@openshift-ci openshift-ci Bot added the size/L Denotes a PR that changes 100-499 lines, ignoring generated files. label Sep 15, 2026
@abhatt-rh abhatt-rh changed the title Administer map Administer and Disconnected map Sep 15, 2026
@openshift-ci openshift-ci Bot added size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. and removed size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Sep 15, 2026
@abhatt-rh

Copy link
Copy Markdown
Contributor Author

/retest

@ocpdocs-previewbot

Copy link
Copy Markdown

🤖 Tue Sep 15 17:35:36 - Prow CI generated the docs preview:

https://119840--ocpdocs-pr.netlify.app/


include::modules/oc-mirror-workflows-fully-disconnected-v2.adoc[leveloffset=+1,toc="no"]

include::modules/oc-mirror-IDMS-ITMS-about.adoc[leveloffset=+1,toc="no"]

Copy link
Copy Markdown
Collaborator

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

🤖 [error] AsciiDocDITA.MapFilename: Use only lowercase alphanumeric characters and hyphens (a-z, 0-9, and -) in file names.

@openshift-ci

openshift-ci Bot commented Sep 15, 2026

Copy link
Copy Markdown

@abhatt-rh: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

size/XL Denotes a PR that changes 500-999 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants