Conversation
SPLAT-2745: Changed vSphere cloud config to be read from openshift-config-managed (cherry picked from commit 7f40a74)
SPLAT-2790: Added new VAP for vSphere infra validation against current cpms and machines (cherry picked from commit 068c67c)
SPLAT-2813: Create VAP E2Es (cherry picked from commit 417ad2f)
SPLAT-2858: Fixing test issue where error result was for other resource type (cherry picked from commit a19340b)
SPLAT-2854: Fixing issue related to unintended VAP interactions (cherry picked from commit 3be0a58)
OCPBUGS-99903: Fixing test issue where error result was for other resource type (cherry picked from commit aadf54d)
SPLAT-2826: Compare against oldObject in vSphere failure-domain VAPs (cherry picked from commit 08e3154)
|
Pipeline controller notification For optional jobs, comment This repository is configured in: LGTM mode |
|
@vr4manta: This pull request references SPLAT-2867 which is a valid jira issue. DetailsIn response to this:
Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the openshift-eng/jira-lifecycle-plugin repository. |
|
Skipping CI for Draft Pull Request. |
|
Important Review skippedAuto reviews are disabled on base/target branches other than the default branch. Please check the settings in the CodeRabbit UI or the ⚙️ Run configurationConfiguration used: Repository: openshift/coderabbit/.coderabbit.yaml Review profile: CHILL Plan: Enterprise Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Comment |
|
/test all |
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
@vr4manta: all tests passed! Full PR test history. Your PR dashboard. DetailsInstructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here. |
SPLAT-2867
Changes
Notes:
This PR is backporting the following PRs in relation to the vSphere multi vCenter day 2 support feature:
Since release 4.20 did not have the original feature gate, this PR require changes to go.mod to pick up the new feature gate.