Skip to content

Update faas-console-plugin to 98368b0 - #315

Open
red-hat-konflux-kflux-prd-rh02[bot] wants to merge 1 commit into
bundle-mainfrom
konflux/component-updates/functions-operator-bundle-component-update-faas-console-plugin
Open

Update faas-console-plugin to 98368b0#315
red-hat-konflux-kflux-prd-rh02[bot] wants to merge 1 commit into
bundle-mainfrom
konflux/component-updates/functions-operator-bundle-component-update-faas-console-plugin

Conversation

@red-hat-konflux-kflux-prd-rh02

@red-hat-konflux-kflux-prd-rh02 red-hat-konflux-kflux-prd-rh02 Bot commented Sep 11, 2026

Copy link
Copy Markdown

Image created from 'https://github.com/openshift/faas-console-plugin?rev=8b8bfb1b0468edfbadb2db316ee4053b8a27f589'

This PR contains the following updates:

Package Update Change
quay.io/redhat-user-workloads/ocp-serverless-tenant/faas-console-plugin digest 2a92227 -> 98368b0

Configuration

📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test
This is Konflux nudging PR, not Mintmaker.

@red-hat-konflux-kflux-prd-rh02 red-hat-konflux-kflux-prd-rh02 Bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. konflux-nudge lgtm Indicates that a PR is ready to be merged. labels Sep 11, 2026
@openshift-ci openshift-ci Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 11, 2026
@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by: red-hat-konflux-kflux-prd-rh02[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

1 similar comment
@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by: red-hat-konflux-kflux-prd-rh02[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD caf041e and 2 for PR HEAD 46597ab in total

Image created from 'https://github.com/openshift/faas-console-plugin?rev=8b8bfb1b0468edfbadb2db316ee4053b8a27f589'

Signed-off-by: red-hat-konflux-kflux-prd-rh02 <190377777+red-hat-konflux-kflux-prd-rh02[bot]@users.noreply.github.com>
@red-hat-konflux-kflux-prd-rh02 red-hat-konflux-kflux-prd-rh02 Bot changed the title Update faas-console-plugin to d3b5eb8 Update faas-console-plugin to 98368b0 Sep 12, 2026
@red-hat-konflux-kflux-prd-rh02
red-hat-konflux-kflux-prd-rh02 Bot force-pushed the konflux/component-updates/functions-operator-bundle-component-update-faas-console-plugin branch from 46597ab to 3354e51 Compare September 12, 2026 09:16
@openshift-ci openshift-ci Bot removed the lgtm Indicates that a PR is ready to be merged. label Sep 12, 2026
@openshift-ci

openshift-ci Bot commented Sep 12, 2026

Copy link
Copy Markdown

New changes are detected. LGTM label has been removed.

@openshift-ci

openshift-ci Bot commented Sep 12, 2026

Copy link
Copy Markdown

@red-hat-konflux-kflux-prd-rh02[bot]: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-install-aws 3354e51 link true /test e2e-install-aws

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. konflux-nudge size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants