Skip to content

Update functions-operator-bundle to a6f9c7e - #313

Open
red-hat-konflux-kflux-prd-rh02[bot] wants to merge 1 commit into
bundle-mainfrom
konflux/component-updates/serverless-functions-fbc-422-component-update-functions-operator-bundle
Open

Update functions-operator-bundle to a6f9c7e#313
red-hat-konflux-kflux-prd-rh02[bot] wants to merge 1 commit into
bundle-mainfrom
konflux/component-updates/serverless-functions-fbc-422-component-update-functions-operator-bundle

Conversation

@red-hat-konflux-kflux-prd-rh02

Copy link
Copy Markdown

Image created from 'https://github.com/openshift/functions-operator?rev=caf041e7fbcaf3d086bbafcbddf77e26f51e4b01'

This PR contains the following updates:

Package Update Change
quay.io/redhat-user-workloads/ocp-serverless-tenant/functions-operator-bundle digest 994f71b -> a6f9c7e

Configuration

📅 Schedule: Branch creation - At any time (no schedule defined), Automerge - At any time (no schedule defined).

🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.

Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

👻 Immortal: This PR will be recreated if closed unmerged. Get config help if that's undesired.


  • If you want to rebase/retry this PR, check this box

To execute skipped test pipelines write comment /ok-to-test
This is Konflux nudging PR, not Mintmaker.

Image created from 'https://github.com/openshift/functions-operator?rev=caf041e7fbcaf3d086bbafcbddf77e26f51e4b01'

Signed-off-by: red-hat-konflux-kflux-prd-rh02 <190377777+red-hat-konflux-kflux-prd-rh02[bot]@users.noreply.github.com>
@red-hat-konflux-kflux-prd-rh02 red-hat-konflux-kflux-prd-rh02 Bot added approved Indicates a PR has been approved by an approver from all required OWNERS files. konflux-nudge lgtm Indicates that a PR is ready to be merged. labels Sep 11, 2026
@openshift-ci openshift-ci Bot added the size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. label Sep 11, 2026
@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by: red-hat-konflux-kflux-prd-rh02[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

1 similar comment
@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is APPROVED

Approval requirements bypassed by manually added approval.

This pull-request has been approved by: red-hat-konflux-kflux-prd-rh02[bot]

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown

@red-hat-konflux-kflux-prd-rh02[bot]: The following test failed, say /retest to rerun all failed tests or /retest-required to rerun all mandatory failed tests:

Test name Commit Details Required Rerun command
ci/prow/e2e-install-aws 2c354c4 link true /test e2e-install-aws

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown

/retest-required

Remaining retests: 0 against base HEAD caf041e and 2 for PR HEAD 2c354c4 in total

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. konflux-nudge lgtm Indicates that a PR is ready to be merged. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

0 participants