Skip to content

build: add mandatory Dockerfile labels - #190

Open
maschmid wants to merge 1 commit into
openshift:masterfrom
maschmid:maschmid-add-mandatory-dockerfile-labels
Open

build: add mandatory Dockerfile labels#190
maschmid wants to merge 1 commit into
openshift:masterfrom
maschmid:maschmid-add-mandatory-dockerfile-labels

Conversation

@maschmid

Copy link
Copy Markdown
  • 🔧 Add mandatory Dockerfile labels

@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown

Pipeline controller notification
This repo is configured to use the pipeline controller. Second-stage tests will be triggered either automatically or after lgtm label is added, depending on the repository configuration. The pipeline controller will automatically detect which contexts are required and will utilize /test Prow commands to trigger the second stage.

For optional jobs, comment /test ? to see a list of all defined jobs. To trigger manually all jobs from second stage use /pipeline required command.

This repository is configured in: LGTM mode

@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign cragsmann for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@openshift-ci

openshift-ci Bot commented Sep 11, 2026

Copy link
Copy Markdown

@maschmid: all tests passed!

Full PR test history. Your PR dashboard.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes-sigs/prow repository. I understand the commands that are listed here.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant