Skip to content

Add FBC Release CRs from release-1.37 revision for stage - #1015

Open
serverless-qe wants to merge 1 commit into
openshift-knative:mainfrom
serverless-qe:release-crs-release-1.37-fbc-stage
Open

serverless-qe wants to merge 1 commit into
openshift-knative:mainfrom
serverless-qe:release-crs-release-1.37-fbc-stage

Conversation

@serverless-qe

Copy link
Copy Markdown
Collaborator

Add FBC Release CRs from release-1.37 revision for stage

@openshift-ci

openshift-ci Bot commented Sep 18, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by: serverless-qe
Once this PR has been reviewed and has the lgtm label, please assign aliok for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@serverless-qe
serverless-qe force-pushed the release-crs-release-1.37-fbc-stage branch from 52f8cf7 to af34357 Compare September 18, 2026 09:44
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant