chore(deps): update vueuse monorepo to v15 - #3274
Open
renovate[bot] wants to merge 1 commit into
Open
renovate[bot] wants to merge 1 commit into
renovate[bot] wants to merge 1 commit into
Conversation
Contributor
|
The latest updates on your projects. Learn more about Vercel for GitHub.
1 Skipped Deployment
|
|
Review the following changes in direct dependencies. Learn more about Socket for GitHub.
|
Codecov Report✅ All modified and coverable lines are covered by tests. 📢 Thoughts on this report? Let us know! |
📊 Dependency Size ChangesWarning This PR adds 1.5 MB of new dependencies, which exceeds the threshold of 200 kB.
Total size change: 1.5 MB |
renovate
Bot
force-pushed
the
renovate/major-vueuse-monorepo
branch
from
September 21, 2026 08:52
aef6f0f to
450e9c5
Compare
This branch was successfully deployed
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
14.4.0→15.0.014.4.0→15.0.014.4.0→15.0.0^14.4.0→^15.0.014.4.0→15.0.0Warning
Some dependencies could not be looked up. Check the Dependency Dashboard for more information.
Release Notes
vueuse/vueuse (@vueuse/core)
v15.0.0Compare Source
🚨 Breaking Changes
scheduler- by @OrbisK in #5564 (c02ab)trailingoption from false to true - by @kyumoon and @OrbisK in #5249 (e9fe3)🚀 Features
🐞 Bug Fixes
🏎 Performance
View changes on GitHub
Configuration
📅 Schedule: (UTC)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about these updates again.
This PR was generated by Mend Renovate. View the repository job log.