Skip to content

Update module sigs.k8s.io/controller-runtime to v0.25.2 - #200

Merged
gardener-prow[bot] merged 1 commit into
mainfrom
renovate/sigs.k8s.io-controller-runtime-0.25.x
Oct 1, 2026
Merged

gardener-prow[bot] merged 1 commit into
mainfrom
renovate/sigs.k8s.io-controller-runtime-0.25.x

Conversation

@gardener-ci-robot

Copy link
Copy Markdown
Contributor

This PR contains the following updates:

Package Change Age Confidence
sigs.k8s.io/controller-runtime v0.25.1 → v0.25.2 age confidence

Release Notes

kubernetes-sigs/controller-runtime (sigs.k8s.io/controller-runtime)

v0.25.2

Compare Source

What's Changed

Full Changelog: kubernetes-sigs/controller-runtime@v0.25.1...v0.25.2


Configuration

📅 Schedule: (UTC)

  • Branch creation
    • At any time (no schedule defined)
  • Automerge
    • At any time (no schedule defined)

🚦 Automerge: Enabled.

♻ Rebasing: Whenever PR is behind base branch, or you tick the rebase/retry checkbox.

🔕 Ignore: Close this PR and you won't be reminded about this update again.


  • If you want to rebase/retry this PR, check this box

This PR has been generated by Mend Renovate CLI.

Signed-off-by: gardener-ci-robot <gardener.ci.robot@gmail.com>
@gardener-ci-robot
gardener-ci-robot requested a review from a team as a code owner October 1, 2026 14:41
@gardener-ci-robot gardener-ci-robot added the kind/enhancement Enhancement, improvement, extension label Oct 1, 2026
@gardener-ci-robot

Copy link
Copy Markdown
Contributor Author

/label skip-review

@gardener-prow

gardener-prow Bot commented Oct 1, 2026

Copy link
Copy Markdown

[APPROVALNOTIFIER] This PR is NOT APPROVED

This pull-request has been approved by:
Once this PR has been reviewed and has the lgtm label, please assign anluerz for approval. For more information see the Code Review Process.

The full list of commands accepted by this bot can be found here.

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@gardener-prow gardener-prow Bot added skip-review Indicates a PR is trusted, used by tide for auto-merging PRs. size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. labels Oct 1, 2026
@gardener-prow
gardener-prow Bot merged commit c2636c4 into main Oct 1, 2026
22 checks passed
@gardener-ci-robot
gardener-ci-robot deleted the renovate/sigs.k8s.io-controller-runtime-0.25.x branch October 1, 2026 15:26
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

kind/enhancement Enhancement, improvement, extension size/XS Denotes a PR that changes 0-9 lines, ignoring generated files. skip-review Indicates a PR is trusted, used by tide for auto-merging PRs.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant