Repository navigation
fix: bump @eclipse-che/license-tool from 2.0.1-next.7742774 to 2.0.1 - #476
dependabot[bot] wants to merge 1 commit into
Conversation
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: dependabot[bot] The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
|
Important Review skippedBot user detected. To trigger a single review, invoke the ⚙️ Run configurationConfiguration used: defaults Review profile: CHILL Plan: Advanced Run ID: You can disable this status message by setting the Use the checkbox below for a quick retry:
Thanks for using CodeRabbit! It's free for OSS, and your support helps us grow. If you like it, consider giving us a shout-out. Comment |
Bumps [@eclipse-che/license-tool](https://github.com/che-incubator/dash-licenses) from 2.0.1-next.7742774 to 2.0.1. - [Commits](https://github.com/che-incubator/dash-licenses/commits) --- updated-dependencies: - dependency-name: "@eclipse-che/license-tool" dependency-version: 2.0.1 dependency-type: direct:development update-type: version-update:semver-patch ... Signed-off-by: dependabot[bot] <support@github.com>
bdf9899 to
d4e4438
Compare
Bumps @eclipse-che/license-tool from 2.0.1-next.7742774 to 2.0.1.
Commits