chore(deps): update renovate/renovate docker tag to v44.115.10 - #759
Open
renovate[bot] wants to merge 1 commit into
Open
renovate[bot] wants to merge 1 commit into
renovate[bot] wants to merge 1 commit into
Conversation
Contributor
|
[APPROVALNOTIFIER] This PR is NOT APPROVED This pull-request has been approved by: The full list of commands accepted by this bot can be found here. DetailsNeeds approval from an approver in each of these files:Approvers can indicate their approval by writing |
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 27, 2026 16:22
5d8d030 to
d08b9f0
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 27, 2026 20:31
d08b9f0 to
d4d0638
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 28, 2026 00:22
d4d0638 to
7df9d1d
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 28, 2026 04:42
7df9d1d to
b6dbbc8
Compare
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 28, 2026 21:14
b6dbbc8 to
2708ae7
Compare
Signed-off-by: renovate[bot] <29139614+renovate[bot]@users.noreply.github.com>
renovate
Bot
force-pushed
the
renovate/renovate-renovate-44.x
branch
from
September 29, 2026 02:54
2708ae7 to
50c11f7
Compare
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
This PR contains the following updates:
44.112.0→44.115.1044.119.1(+12)Release Notes
renovatebot/renovate (renovate/renovate)
v44.115.10Compare Source
Build System
v44.115.9Compare Source
Bug Fixes
v44.115.8Compare Source
v44.115.7Compare Source
Bug Fixes
v44.115.6Compare Source
v44.115.5Compare Source
Bug Fixes
Documentation
Miscellaneous Chores
v44.115.4Compare Source
Miscellaneous Chores
Build System
v44.115.3Compare Source
Bug Fixes
Miscellaneous Chores
Code Refactoring
v44.115.2Compare Source
Bug Fixes
Miscellaneous Chores
v44.115.1Compare Source
Code Refactoring
Build System
v44.115.0Compare Source
Features
v44.114.1Compare Source
Bug Fixes
v44.114.0Compare Source
Features
v44.113.0Compare Source
Features
Miscellaneous Chores
v44.112.3Compare Source
Bug Fixes
Miscellaneous Chores
v44.112.2Compare Source
Bug Fixes
Miscellaneous Chores
Continuous Integration
v44.112.1Compare Source
Bug Fixes
Configuration
📅 Schedule: (in timezone Europe/London)
🚦 Automerge: Disabled by config. Please merge this manually once you are satisfied.
♻ Rebasing: Whenever PR becomes conflicted, or you tick the rebase/retry checkbox.
🔕 Ignore: Close this PR and you won't be reminded about this update again.
This PR was generated by Mend Renovate. View the repository job log.